Repository navigation
Bump the go-security-updates group across 3 directories with 2 updates - #2759
dependabot[bot] wants to merge 1 commit into
Conversation
|
Hi @dependabot[bot]. Thanks for your PR. I'm waiting for a GoogleCloudPlatform member to verify that this patch is reasonable to test. If it is, they should reply with Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
Bumps the go-security-updates group with 1 update in the /common directory: [go.opentelemetry.io/otel/sdk](https://github.com/open-telemetry/opentelemetry-go). Bumps the go-security-updates group with 1 update in the /gce_image_publish directory: [go.opentelemetry.io/otel/sdk](https://github.com/open-telemetry/opentelemetry-go). Bumps the go-security-updates group with 1 update in the /proto/go directory: google.golang.org/protobuf. Updates `go.opentelemetry.io/otel/sdk` from 1.44.0 to 1.45.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.44.0...v1.45.0) Updates `go.opentelemetry.io/otel/sdk` from 1.44.0 to 1.45.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.44.0...v1.45.0) Updates `go.opentelemetry.io/otel/sdk` from 1.44.0 to 1.45.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.44.0...v1.45.0) Updates `go.opentelemetry.io/otel/sdk` from 1.44.0 to 1.45.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.44.0...v1.45.0) Updates `go.opentelemetry.io/otel/sdk` from 1.44.0 to 1.45.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.44.0...v1.45.0) Updates `go.opentelemetry.io/otel/sdk` from 1.44.0 to 1.45.0 - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.44.0...v1.45.0) Updates `google.golang.org/protobuf` from 1.25.0 to 1.33.0 Updates `google.golang.org/protobuf` from 1.25.0 to 1.33.0 Updates `google.golang.org/protobuf` from 1.25.0 to 1.33.0 Updates `google.golang.org/protobuf` from 1.25.0 to 1.33.0 --- updated-dependencies: - dependency-name: go.opentelemetry.io/otel/sdk dependency-version: 1.45.0 dependency-type: indirect dependency-group: go-security-updates - dependency-name: go.opentelemetry.io/otel/sdk dependency-version: 1.45.0 dependency-type: indirect dependency-group: go-security-updates - dependency-name: google.golang.org/protobuf dependency-version: 1.33.0 dependency-type: direct:production dependency-group: go-security-updates ... Signed-off-by: dependabot[bot] <support@github.com>
0c3ede8 to
6dc0be9
Compare
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
Rebasing might not happen immediately, so don't worry if this takes some time.
Note: if you make any changes to this PR yourself, they will take precedence over the rebase.
Bumps the go-security-updates group with 1 update in the /common directory: go.opentelemetry.io/otel/sdk.
Bumps the go-security-updates group with 1 update in the /gce_image_publish directory: go.opentelemetry.io/otel/sdk.
Bumps the go-security-updates group with 1 update in the /proto/go directory: google.golang.org/protobuf.
Updates
go.opentelemetry.io/otel/sdkfrom 1.44.0 to 1.45.0Release notes
Sourced from go.opentelemetry.io/otel/sdk's releases.
... (truncated)
Changelog
Sourced from go.opentelemetry.io/otel/sdk's changelog.
... (truncated)
Commits
93a693eRelease v1.45.0 (#8693)c65d435Merge commit from fork223f9fdsdk/metric: remove obsolete randomFloat64 TODO (#8685)06272bcfix(deps): update googleapis to 6ac0973 (#8694)a4f238fchore(deps): update github.com/charmbracelet/ultraviolet digest to 8b69304 (#...37140e7chore(deps): update codspeedhq/action action to v5.0.2 (#8690)cef0855chore(deps): update module github.com/lucasb-eyer/go-colorful to v1.4.1 (#8689)e814a72Merge commit from forkbfd8eb7chore(deps): update github.com/golangci/rowserrcheck digest to d2031e3 (#8687)48db2c6chore(deps): update github/codeql-action action to v4.37.5 (#8692)Updates
go.opentelemetry.io/otel/sdkfrom 1.44.0 to 1.45.0Release notes
Sourced from go.opentelemetry.io/otel/sdk's releases.
... (truncated)
Changelog
Sourced from go.opentelemetry.io/otel/sdk's changelog.
... (truncated)
Commits
93a693eRelease v1.45.0 (#8693)c65d435Merge commit from fork223f9fdsdk/metric: remove obsolete randomFloat64 TODO (#8685)06272bcfix(deps): update googleapis to 6ac0973 (#8694)a4f238fchore(deps): update github.com/charmbracelet/ultraviolet digest to 8b69304 (#...37140e7chore(deps): update codspeedhq/action action to v5.0.2 (#8690)cef0855chore(deps): update module github.com/lucasb-eyer/go-colorful to v1.4.1 (#8689)e814a72Merge commit from forkbfd8eb7chore(deps): update github.com/golangci/rowserrcheck digest to d2031e3 (#8687)48db2c6chore(deps): update github/codeql-action action to v4.37.5 (#8692)Updates
go.opentelemetry.io/otel/sdkfrom 1.44.0 to 1.45.0Release notes
Sourced from go.opentelemetry.io/otel/sdk's releases.
... (truncated)
Changelog
Sourced from go.opentelemetry.io/otel/sdk's changelog.