Skip to content

Update OWNERS - #2760

Merged
copybara-service[bot] merged 1 commit into
GoogleCloudPlatform:masterfrom
drewhli:owners
Sep 30, 2026
Merged

copybara-service[bot] merged 1 commit into
GoogleCloudPlatform:masterfrom
drewhli:owners

Conversation

@drewhli

@drewhli drewhli commented Sep 30, 2026

Copy link
Copy Markdown
Contributor

@google-oss-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: drewhli

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@copybara-service
copybara-service Bot merged commit ab2c377 into GoogleCloudPlatform:master Sep 30, 2026
5 of 6 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant