Repository navigation
Bump cloud.google.com/go/compute/metadata in /cli_tools - #2762
dependabot[bot] wants to merge 1 commit into
Conversation
|
Hi @dependabot[bot]. Thanks for your PR. I'm waiting for a GoogleCloudPlatform member to verify that this patch is reasonable to test. If it is, they should reply with Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
Bumps [cloud.google.com/go/compute/metadata](https://github.com/googleapis/google-cloud-go) in `/cli_tools` from 0.9.1 to 0.10.0. Updates `cloud.google.com/go/compute/metadata` from 0.9.1 to 0.10.0 - [Release notes](https://github.com/googleapis/google-cloud-go/releases) - [Changelog](https://github.com/googleapis/google-cloud-go/blob/main/CHANGES.md) - [Commits](googleapis/google-cloud-go@auth/v0.9.1...v0.10.0) Updates `cloud.google.com/go/compute/metadata` from 0.9.1 to 0.10.0 - [Release notes](https://github.com/googleapis/google-cloud-go/releases) - [Changelog](https://github.com/googleapis/google-cloud-go/blob/main/CHANGES.md) - [Commits](googleapis/google-cloud-go@auth/v0.9.1...v0.10.0) --- updated-dependencies: - dependency-name: cloud.google.com/go/compute/metadata dependency-version: 0.10.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: go-version-updates/cloud.google.com/go/compute/metadata - dependency-name: cloud.google.com/go/compute/metadata dependency-version: 0.10.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: go-version-updates/cloud.google.com/go/compute/metadata ... Signed-off-by: dependabot[bot] <support@github.com>
b4fd48f to
cd68e4a
Compare
Bumps cloud.google.com/go/compute/metadata in
/cli_toolsfrom 0.9.1 to 0.10.0.Updates
cloud.google.com/go/compute/metadatafrom 0.9.1 to 0.10.0Changelog
Sourced from cloud.google.com/go/compute/metadata's changelog.
... (truncated)
Commits
a5913b3pubsub: add Subscription.Update; remove Subcription.ModifyPushConfig1bfb4aastorage: fix reference to CopyFrom in doc comment05b84c5pubsub: add test for ModifyPushConfig328063apubsub: fix data races in stream tests927c812pubsub: pull messages concurrently7c205daadd video intelligence to READMEf6bedb5bigtable: Add InstanceInfo admin method.1a5f48fvision: deprecate manual client1dd9a1espanner: increase code coverage2b81b39translate: mark stableUpdates
cloud.google.com/go/compute/metadatafrom 0.9.1 to 0.10.0Changelog
Sourced from cloud.google.com/go/compute/metadata's changelog.
... (truncated)
Commits
a5913b3pubsub: add Subscription.Update; remove Subcription.ModifyPushConfig1bfb4aastorage: fix reference to CopyFrom in doc comment05b84c5pubsub: add test for ModifyPushConfig328063apubsub: fix data races in stream tests927c812pubsub: pull messages concurrently7c205daadd video intelligence to READMEf6bedb5bigtable: Add InstanceInfo admin method.1a5f48fvision: deprecate manual client1dd9a1espanner: increase code coverage2b81b39translate: mark stable