Skip to content

Bump github.com/golang/mock in /proto/go - #2763

Open
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/go_modules/proto/go/go_modules-7502c235b3
Open

dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/go_modules/proto/go/go_modules-7502c235b3

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Oct 6, 2026

Copy link
Copy Markdown
Contributor

Bumps github.com/golang/mock in /proto/go from 1.1.1 to 1.6.0.

Updates github.com/golang/mock from 1.1.1 to 1.6.0

Release notes

Sourced from github.com/golang/mock's releases.

v1.6.0

Changelog

317c030 Best effort guesses for output package path (#547) c59ba11 add ARM to support apple M1 chip to releaser (#562) 58935d8 add a basic CONTRIBUTING.md (#535) a5582fc add docs on 1.16 install and adding to PATH (#534) 0cd3aaf add flags documentation (#539) 64b0b80 add notes and error helper for vendor+reflect error (#567) e303461 add type information to error messages (#559) 0cdccf5 feat add InAnyOrder matcher (#546) 82ce4a7 feat validate Do & DoReturn args (#558) 93308c3 fix broken badge (#525) 9336b7e fix error message in parse.go (#540) ab03293 fix ill-formatted message with fmt-verbs like %s (#564) bb5fd5e fix linter errors (#552) aba2ff9 fix parse array with the external const correctly (#569) 6ff1070 fix parse arrays with const length correctly (#520) 7f5f64d fixup some docs and templates (#524) 7078515 refactor go:generate lines so they are consistently placed (#527) 7105dde refactor mockgen and cleanup (#536) f36d14a test(sample/user_test.go): minor correction at t.Errorf (#544) ef4ad87 update CI for 1.16 (#526) ad820b0 update Finish docs for Go1.14+ (#556) 2421472 update dependencies (#528) 953a5bb update user mock to be in test package (#566) d19a212 upgrade dependencies (#557)

v1.5.0

Changelog

0b87a54 Add a period to the end of comments (#414) d2fe5cd Add example for Call.Do and Call.DoAndReturn (#470) 91d4b5c Add tests for various Do/DoAndReturn calls (#430) 92f53b0 Avoid using packages.Load (#420) aff3767 Fix empty error message when call is exhausted (#460) 8734ec5 Format generated files with goimports (#458) 69e02d3 MOCK-429: add support for assignable types to Eq matcher (#481) d476d65 Parse parenthesized parameter-type. (#421) ccaa079 Using pacakges.NeedName (#418) f67ce0c add CODEOWNERS file (#522) 6d816de add default calling of ctrl.Finish() in go1.14+ (#422) 7b53c4d check error for os.Setenv in parse_test.go (#472) b9a8584 deduplicate methods to allow overlapping methods on embedded interfaces (#498) 10192bd fix 1.11 ga test failure (#511) dcd893e fix Test_createPackageMap for 1.15 (#512) 44e6f1e fix docs for AssignableToTypeOf (#452) a23c5e7 fix issues related to source package imports (#507) 11d9cab fix readme docs on removing Finish calls (#461) 0f6dc21 format variadic arguments with GotFormatter (#434)

... (truncated)

Commits

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore <dependency name> major version will close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)
  • @dependabot ignore <dependency name> minor version will close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)
  • @dependabot ignore <dependency name> will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)
  • @dependabot unignore <dependency name> will remove all of the ignore conditions of the specified dependency
  • @dependabot unignore <dependency name> <ignore condition> will remove the ignore condition of the specified dependency and ignore conditions

Bumps [github.com/golang/mock](https://github.com/golang/mock) in `/proto/go` from 1.1.1 to 1.6.0.


Updates `github.com/golang/mock` from 1.1.1 to 1.6.0
- [Release notes](https://github.com/golang/mock/releases)
- [Commits](golang/mock@v1.1.1...v1.6.0)

---
updated-dependencies:
- dependency-name: github.com/golang/mock
  dependency-version: 1.6.0
  dependency-type: direct:production
  update-type: version-update:semver-minor
  dependency-group: go_modules
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file go Pull requests that update go code labels Oct 6, 2026
@google-oss-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: dependabot[bot]
Once this PR has been reviewed and has the lgtm label, please assign christiec17 for approval. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@google-oss-prow

Copy link
Copy Markdown

Hi @dependabot[bot]. Thanks for your PR.

I'm waiting for a GoogleCloudPlatform member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file go Pull requests that update go code needs-ok-to-test size/S

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants