Skip to content

fix: exclude transform-typescript when present - #1265

Merged
jbroma merged 3 commits into
mainfrom
fix/exclude-transform-typescript
Aug 21, 2025
Merged

jbroma merged 3 commits into
mainfrom
fix/exclude-transform-typescript

Conversation

@jbroma

@jbroma jbroma commented Aug 21, 2025 •

Copy link
Copy Markdown
Contributor

Summary

Exclude transform-typescript from babel rules when detected through babel-swc-loader

Test plan

  • - testers work

@changeset-bot

changeset-bot Bot commented Aug 21, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c95b54b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 6 packages
Name Type
@callstack/repack Patch
@callstack/repack-plugin-expo-modules Patch
@callstack/repack-plugin-nativewind Patch
@callstack/repack-plugin-reanimated Patch
@callstack/repack-dev-server Patch
@callstack/repack-init Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@vercel

vercel Bot commented Aug 21, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Preview Comments Updated (UTC)
repack-website Ready Ready Preview Comment Aug 21, 2025 5:28pm

@jbroma
jbroma merged commit b28d2b2 into main Aug 21, 2025
@jbroma
jbroma deleted the fix/exclude-transform-typescript branch August 21, 2025 17:35

This branch was successfully deployed

1 active deployment
Preview — c95b54b0 Deployed Aug 21, 2025 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant