Skip to content

Chore(deps): bump js-yaml in /modules/web - #8302

Merged
kubermatic-bot merged 1 commit into
mainfrom
dependabot/npm_and_yarn/modules/web/multi-c5ebbb4833
Sep 22, 2026
Merged

kubermatic-bot merged 1 commit into
mainfrom
dependabot/npm_and_yarn/modules/web/multi-c5ebbb4833

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 10, 2026

Copy link
Copy Markdown
Contributor

Bumps js-yaml to 4.3.2 and updates ancestor dependency . These dependencies need to be updated together.

Updates js-yaml from 4.3.1 to 4.3.2

Changelog

Sourced from js-yaml's changelog.

4.3.2 - 2026-08-26

Changed

  • [backport] Hard-limit merge sequence size to 100.

Security

  • [backport] Count empty mappings in merge sequences toward maxTotalMergeKeys to limit CPU usage, #797.
Commits

Updates js-yaml from 3.15.1 to 3.15.2

Changelog

Sourced from js-yaml's changelog.

4.3.2 - 2026-08-26

Changed

  • [backport] Hard-limit merge sequence size to 100.

Security

  • [backport] Count empty mappings in merge sequences toward maxTotalMergeKeys to limit CPU usage, #797.
Commits

@dependabot dependabot Bot added dependencies Pull requests that update a dependency file docs/none Denotes a PR that doesn't need documentation (changes). ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesn't merit a release note. test-issue/none Denotes a PR that doesn't need a test issue (changes). labels Sep 10, 2026
@kubermatic-bot kubermatic-bot added dco-signoff: yes Denotes that all commits in the pull request have the valid DCO signoff message. do-not-merge/docs-needed Indicates that a PR should not merge because it's missing one of the documentation labels. do-not-merge/test-issue-needed Indicates that a PR should not merge because it's missing one of the test issue labels. and removed docs/none Denotes a PR that doesn't need documentation (changes). test-issue/none Denotes a PR that doesn't need a test issue (changes). labels Sep 10, 2026
@kubermatic-bot

Copy link
Copy Markdown
Contributor

Hi @dependabot[bot]. Thanks for your PR.

I'm waiting for a kubermatic member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@kubermatic-bot kubermatic-bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 10, 2026
@codecov

codecov Bot commented Sep 10, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 45.23%. Comparing base (dcb2f2b) to head (5917e13).
⚠️ Report is 1 commits behind head on main.

Additional details and impacted files
@@           Coverage Diff           @@
##             main    #8302   +/-   ##
=======================================
  Coverage   45.23%   45.23%           
=======================================
  Files         449      449           
  Lines       21464    21464           
  Branches     2981     2981           
=======================================
  Hits         9710     9710           
  Misses      11747    11747           
  Partials        7        7           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.
  • 📦 JS Bundle Analysis: Save yourself from yourself by tracking and limiting bundle sizes in JS merges.

@dependabot dependabot Bot changed the title Bump js-yaml in /modules/web Chore(deps): bump js-yaml in /modules/web Sep 10, 2026
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/modules/web/multi-c5ebbb4833 branch 3 times, most recently from d1c86fd to 7157d91 Compare September 16, 2026 11:51
@KhizerRehan

Copy link
Copy Markdown
Contributor

/reopen

Bumps [js-yaml](https://github.com/nodeca/js-yaml) to 4.3.2 and updates ancestor dependency . These dependencies need to be updated together.


Updates `js-yaml` from 4.3.1 to 4.3.2
- [Changelog](https://github.com/nodeca/js-yaml/blob/4.3.2/CHANGELOG.md)
- [Commits](nodeca/js-yaml@4.3.1...4.3.2)

Updates `js-yaml` from 3.15.1 to 3.15.2
- [Changelog](https://github.com/nodeca/js-yaml/blob/4.3.2/CHANGELOG.md)
- [Commits](nodeca/js-yaml@4.3.1...4.3.2)

---
updated-dependencies:
- dependency-name: js-yaml
  dependency-version: 3.15.2
  dependency-type: indirect
- dependency-name: js-yaml
  dependency-version: 4.3.2
  dependency-type: direct:production
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/modules/web/multi-c5ebbb4833 branch from 7157d91 to 5917e13 Compare September 22, 2026 04:44
@KhizerRehan

Copy link
Copy Markdown
Contributor

/docs-none
/test-issue-none
/approve
/lgtm

@kubermatic-bot kubermatic-bot added test-issue/none Denotes a PR that doesn't need a test issue (changes). docs/none Denotes a PR that doesn't need documentation (changes). and removed do-not-merge/test-issue-needed Indicates that a PR should not merge because it's missing one of the test issue labels. do-not-merge/docs-needed Indicates that a PR should not merge because it's missing one of the documentation labels. labels Sep 22, 2026
@kubermatic-bot kubermatic-bot added the lgtm Indicates that a PR is ready to be merged. label Sep 22, 2026
@kubermatic-bot

Copy link
Copy Markdown
Contributor

LGTM label has been added.

DetailsGit tree hash: 61919fa396301361db49211bcc8733cd6d372d08

@kubermatic-bot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: KhizerRehan

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubermatic-bot kubermatic-bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 22, 2026
@kubermatic-bot
kubermatic-bot merged commit 18a3270 into main Sep 22, 2026
10 checks passed
@kubermatic-bot kubermatic-bot added this to the KKP 2.31 milestone Sep 22, 2026
@kubermatic-bot
kubermatic-bot deleted the dependabot/npm_and_yarn/modules/web/multi-c5ebbb4833 branch September 22, 2026 06:29
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. dco-signoff: yes Denotes that all commits in the pull request have the valid DCO signoff message. dependencies Pull requests that update a dependency file docs/none Denotes a PR that doesn't need documentation (changes). lgtm Indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesn't merit a release note. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. test-issue/none Denotes a PR that doesn't need a test issue (changes).

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants