Skip to content

Add tiny-li to sig-docs-zh-reviews - #6527

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:mainfrom
tiny-li:patch-1
Sep 9, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:mainfrom
tiny-li:patch-1

Conversation

@tiny-li

@tiny-li tiny-li commented Aug 4, 2026

Copy link
Copy Markdown
Contributor

I want to join the sig-docs-zh-reviews team.

I have met the requirements for reviewer:

Thanks

@kubernetes-prow kubernetes-prow Bot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Aug 4, 2026
@kubernetes-prow kubernetes-prow Bot added area/github-management Issues or PRs related to GitHub Management subproject sig/docs Categorizes an issue or PR as relevant to SIG Docs. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Aug 4, 2026
@tiny-li

tiny-li commented Aug 4, 2026

Copy link
Copy Markdown
Contributor Author

/cc @my-git9 @windsonsea @tengqm

@my-git9

my-git9 commented Aug 5, 2026

Copy link
Copy Markdown
Member

+1
/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Aug 5, 2026
@windsonsea

Copy link
Copy Markdown
Member

+1
/lgtm

@tiny-li

tiny-li commented Sep 3, 2026

Copy link
Copy Markdown
Contributor Author

/cc @kubernetes/sig-docs-leads

@kubernetes-prow

Copy link
Copy Markdown
Contributor

@tiny-li: GitHub didn't allow me to request PR reviews from the following users: kubernetes/sig-docs-leads.

Note that only kubernetes members and repo collaborators can review this PR, and authors cannot review their own PRs.

Details

In response to this:

/cc @kubernetes/sig-docs-leads

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@natalisucks

Copy link
Copy Markdown

thanks @tiny-li! i can see +1s and LGTM from at least two other reviewers on the team, so from SIG Docs leads i will approve
welcome to the zh-reviews group! 🚀
/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: natalisucks, tiny-li

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 9, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 83b06f7 into kubernetes:main Sep 9, 2026
4 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/github-management Issues or PRs related to GitHub Management subproject cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. sig/docs Categorizes an issue or PR as relevant to SIG Docs. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants