Skip to content

openstack-k8s-operators dependency bump branch: 18.0-fr7 - #1015

Merged
openshift-merge-bot[bot] merged 1 commit into
18.0-fr7from
openstack-dependency-bump/18.0-fr7
Oct 5, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
18.0-fr7from
openstack-dependency-bump/18.0-fr7

Conversation

@github-actions

Copy link
Copy Markdown

Automated changes by create-pull-request GitHub action

@coderabbitai

coderabbitai Bot commented Sep 26, 2026 •

Copy link
Copy Markdown

Important

Review skipped

Bot user detected.

To trigger a single review, invoke the @coderabbitai review command.

⚙️ Run configuration
  • Configuration used: Central YAML (base), Organization UI (inherited)
  • Review profile: CHILL
  • Plan: Advanced
  • Run ID: a2432f06-048e-499d-9c74-8b1847894489

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review
  • Autopilot · Keep fixing CodeRabbit findings and required CI, and resolving merge conflicts

Comment @coderabbitai help to get the list of available commands.

@openshift-ci
openshift-ci Bot requested review from fmount and fultonj September 26, 2026 02:09
@openshift-ci

openshift-ci Bot commented Sep 26, 2026

Copy link
Copy Markdown
Contributor

Hi @github-actions[bot]. Thanks for your PR.

I'm waiting for a openstack-k8s-operators member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c8ac24fc9c9d43dfacbc88abcc862d21

✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 55m 10s
✔️ glance-operator-kuttl SUCCESS in 46m 54s
❌ glance-operator-tempest FAILURE in 1h 36m 42s

@fmount

fmount commented Sep 26, 2026

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/38ade5f4cc454597ba9e55ae2963c32e

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 33m 50s
✔️ glance-operator-kuttl SUCCESS in 44m 34s
❌ glance-operator-tempest FAILURE in 1h 36m 09s

@fmount

fmount commented Sep 27, 2026

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/f87cd353ac514b74a34faafaec13bd9e

✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 59m 06s
✔️ glance-operator-kuttl SUCCESS in 46m 09s
❌ glance-operator-tempest FAILURE in 1h 39m 21s

@stuggi

stuggi commented Sep 28, 2026

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/211652f100e3498dbfee553369d986db

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 58m 40s
✔️ glance-operator-kuttl SUCCESS in 44m 23s
❌ glance-operator-tempest FAILURE in 1h 34m 59s

@fmount

fmount commented Sep 28, 2026 •

Copy link
Copy Markdown
Contributor

The failure I see here is related to the fact that openstack services are unable to connect to Ceph.
I don't see [2] being part of FR7, and iirc we had this issue in the past but we're seeing it now that we branched everything on FR7.

I'm proposing the backport and that patch should fix this issue [3].

[1] https://github.com/openstack-k8s-operators/ci-framework/tree/18.0-fr7
[2] openstack-k8s-operators/ci-framework#4144
[3] openstack-k8s-operators/ci-framework#4209

@github-actions
github-actions Bot force-pushed the openstack-dependency-bump/18.0-fr7 branch from b16f662 to 16bba66 Compare October 3, 2026 02:10
@centosinfra-prod-github-app

Copy link
Copy Markdown

@fmount

fmount commented Oct 3, 2026

Copy link
Copy Markdown
Contributor

/retest

@fmount fmount left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci

openshift-ci Bot commented Oct 3, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: fmount, github-actions[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved label Oct 3, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit 76c799c into 18.0-fr7 Oct 5, 2026
8 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants